Skip to content

chore: add CODEOWNERS for default review assignment - #378

Merged
ZhiyuCircle merged 2 commits into
circlefin:mainfrom
melekes:chore-add-codeowners
Sep 11, 2026
Merged

chore: add CODEOWNERS for default review assignment#378
ZhiyuCircle merged 2 commits into
circlefin:mainfrom
melekes:chore-add-codeowners

Conversation

@melekes

@melekes melekes commented Sep 11, 2026

Copy link
Copy Markdown
Contributor

No description provided.

@ZhiyuCircle
ZhiyuCircle merged commit 3f1efe9 into circlefin:main Sep 11, 2026
5 checks passed
@github-actions github-actions Bot added the pending-import Merged PR awaiting reverse-sync to upstream label Sep 11, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

pending-import Merged PR awaiting reverse-sync to upstream

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants