Conversation
|
The latest updates on your projects. Learn more about Vercel for GitHub.
1 Skipped Deployment
|
adinauer
approved these changes
Sep 29, 2026
adinauer
left a comment
Member
There was a problem hiding this comment.
Mostly LGTM, just one comment
| ApolloClient apollo = SentryApolloBuilderExtensionsKt.sentryTracing( | ||
| new ApolloClient.Builder(), | ||
| true, | ||
| Collections.singletonList(SentryOptions.DEFAULT_PROPAGATION_TARGETS), |
Member
There was a problem hiding this comment.
m: DEFAULT_PROPAGATION_TARGETS is marked internal. We could just provide a custom list as sample here or hard code the .* value. Or we open up the default constant.
Collaborator
Author
There was a problem hiding this comment.
Actually, i think i'll remove it. Since we don't show that parameter in the apollo-4 docs or in the kotlin version of this one. WDYT?
Collaborator
Author
There was a problem hiding this comment.
actually, that is an issue in the apollo-4 docs. have a list of issues with the apollo3 and 4 docs, that get their own issue and pr. Went with ".*"
This branch was successfully deployed
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
DESCRIBE YOUR PR
Docs for the apollo5 integration implemented in getsentry/sentry-java#6074
IS YOUR CHANGE URGENT?
Help us prioritize incoming PRs by letting us know when the change needs to go live.
Select exactly one option. For deadlines, replace
YYYY-MM-DDwith the due date. You can update this information later by editing the PR description.