Skip to content

translate/reference/useState - #651

Merged
deblasis merged 1 commit into
translate/reference/useReffrom
translate/reference/useState
Sep 14, 2026
Merged

deblasis merged 1 commit into
translate/reference/useReffrom
translate/reference/useState

Conversation

@deblasis

Copy link
Copy Markdown
Collaborator

No description provided.

@deblasis
deblasis added this pull request to stack #640 September 14, 2026 10:12
@deblasis
deblasis force-pushed the translate/reference/useState branch from 2b77b62 to 3b6faf6 Compare September 14, 2026 10:12
@vercel

vercel Bot commented Sep 14, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated
legacy-it-reactjs-org Ignored Ignored Preview Sep 14, 2026 7:48pm UTC

Request Review

@github-actions

github-actions Bot commented Sep 14, 2026

Copy link
Copy Markdown

Size changes

Details

📦 Next.js Bundle Analysis for react-dev

This analysis was generated by the Next.js Bundle Analysis action. 🤖

This PR introduced no changes to the JavaScript bundle! 🙌

@deblasis
deblasis force-pushed the translate/reference/useState branch from 3b6faf6 to 1d8dc73 Compare September 14, 2026 15:12
@deblasis
deblasis marked this pull request as ready for review September 14, 2026 15:15
deblasis added a commit that referenced this pull request Sep 14, 2026
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
deblasis added a commit that referenced this pull request Sep 14, 2026
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
@deblasis
deblasis force-pushed the translate/reference/useState branch from 2cd119d to 6c8dc18 Compare September 14, 2026 17:40
deblasis added a commit that referenced this pull request Sep 14, 2026
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
@deblasis
deblasis force-pushed the translate/reference/useState branch from 6c8dc18 to 7355dfc Compare September 14, 2026 18:31
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
@deblasis
deblasis force-pushed the translate/reference/useState branch from 7355dfc to b2da603 Compare September 14, 2026 19:48
@deblasis
deblasis merged commit 0c6d71a into main Sep 14, 2026
10 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant