Skip to content

translate/reference/react dom common - #663

Draft
deblasis wants to merge 1 commit into
translate/reference/startTransitionfrom
translate/reference/react-dom-common
Draft

translate/reference/react dom common#663
deblasis wants to merge 1 commit into
translate/reference/startTransitionfrom
translate/reference/react-dom-common

Conversation

@deblasis

Copy link
Copy Markdown
Collaborator

No description provided.

@deblasis
deblasis added this pull request to stack #640 September 14, 2026 10:46
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
@deblasis
deblasis force-pushed the translate/reference/react-dom-common branch from aa355a0 to 4c40869 Compare September 14, 2026 10:46
@vercel

vercel Bot commented Sep 14, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated
legacy-it-reactjs-org Ignored Ignored Preview Sep 14, 2026 10:47am UTC

Request Review

@github-actions

github-actions Bot commented Sep 14, 2026

Copy link
Copy Markdown

Size changes

Details

📦 Next.js Bundle Analysis for react-dev

This analysis was generated by the Next.js Bundle Analysis action. 🤖

One Page Changed Size

The following page changed size from the code in this PR compared to its base branch:

Page Size (compressed) First Load
/[[...markdownPath]] 130.7 KB (🟡 +6 B) 245.85 KB
Details

Only the gzipped size is provided here based on an expert tip.

First Load is the size of the global bundle plus the bundle for the individual page. If a user were to show up to your website and land on a given page, the first load size represents the amount of javascript that user would need to download. If next/link is used, subsequent page loads would only need to download that page's bundle (the number in the "Size" column), since the global bundle has already been downloaded.

Any third party scripts you have added directly to your app using the <script> tag are not accounted for in this analysis

Next to the size is how much the size has increased or decreased compared with the base branch of this PR. If this percentage has increased by 10% or more, there will be a red status indicator applied, indicating that special attention should be given to this.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant